Skip to content

Defer failure if /boot/efi/EFI doesn't exist - #775

Merged
openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
steveb:defer-efi
Aug 31, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
steveb:defer-efi

Conversation

@steveb

@steveb steveb commented Aug 11, 2026

Copy link
Copy Markdown
Collaborator

Grub and shim files in /boot/efi/EFI won't be required if /var/lib/ironic is already pre-populated from
/usr/share/ironic-operator/var-lib-ironic/, so it should only fail if those files need to be sourced from /boot/efi/EFI.

This won't cause issues with older images, and allows for s2i images which won't have /boot/efi/EFI files available.

Jira: OSPRH-34544

Checklist before requesting a review

  • I have performed a self-review of my code and confirmed it passes tests
  • Performed pre-commit run --all
  • Tested operator image in a test/dev environment. It can be CRC via install_yamls or a hotstack instance (optional)
  • Verified that no failures present in logs(optional):
    • ironic-operator-build-deploy-kuttl
    • podified-multinode-ironic-deployment

@openshift-ci
openshift-ci Bot requested review from dprince and hjensas August 11, 2026 22:38
Grub and shim files in /boot/efi/EFI won't be required if
/var/lib/ironic is already pre-populated from
/usr/share/ironic-operator/var-lib-ironic/, so it should only fail if
those files need to be sourced from /boot/efi/EFI.

This won't cause issues with older images, and allows for s2i images
which won't have /boot/efi/EFI files available.
@coderabbitai

coderabbitai Bot commented Aug 27, 2026

Copy link
Copy Markdown

Important

  • 🔍 Trigger review

This repository does not receive automatic reviews because it has fewer than 10 stars.

⚙️ Run configuration

Configuration used: Central YAML (base), Organization UI (inherited)

Review profile: CHILL

Plan: Pro Plus

Run ID: e0585a30-51e5-4c44-b7da-0eaeccff46b2


Comment @coderabbitai help to get the list of available commands.

@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: juliakreger, steveb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 764d105 into openstack-k8s-operators:main Aug 31, 2026
8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants